Skip to content

openstack-k8s-operators dependency bump branch: main - #2080

Open
github-actions[bot] wants to merge 1 commit into
mainfrom
openstack-dependency-bump/main
Open

github-actions[bot] wants to merge 1 commit into
mainfrom
openstack-dependency-bump/main

Conversation

@github-actions

Copy link
Copy Markdown

Automated changes by create-pull-request GitHub action

@coderabbitai

coderabbitai Bot commented Sep 16, 2026

Copy link
Copy Markdown

Important

Review skipped

Bot user detected.

To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Central YAML (base), Organization UI (inherited)

Review profile: CHILL

Plan: Advanced

Run ID: 2a438bca-7ee8-430a-ad43-53a608d3d0b4

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@openshift-ci
openshift-ci Bot requested review from abays and slagle September 16, 2026 05:50
@openshift-ci

openshift-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

Hi @github-actions[bot]. Thanks for your PR.

I'm waiting for a openstack-k8s-operators member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/29a4a316f6d04dafa8710fcd5d2c1e43

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 33m 38s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 20m 18s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 33m 49s
adoption-standalone-to-crc-ceph-provider FAILURE in 56m 02s
openstack-operator-tempest-multinode FAILURE in 1h 26m 02s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 2h 00m 47s

@stuggi

stuggi commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7b20c82c366d476397d3ad1b0d58df63

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 32m 58s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 24m 38s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 32m 51s
adoption-standalone-to-crc-ceph-provider POST_FAILURE in 3h 03m 32s
openstack-operator-tempest-multinode FAILURE in 1h 30m 31s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 1h 58m 12s

@karelyatin

Copy link
Copy Markdown
Contributor

recheck unrelated failures

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2af06b4b43ba474bb099b99e9b4dfb59

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 26m 50s
podified-multinode-edpm-deployment-crc NODE_FAILURE Node(set) request 099-0000205685 failed in 0s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 33m 30s
✔️ adoption-standalone-to-crc-ceph-provider SUCCESS in 3h 03m 38s
✔️ openstack-operator-tempest-multinode SUCCESS in 1h 47m 15s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 1h 57m 18s

@stuggi

stuggi commented Sep 17, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/4f54f68fa2ec4d54b03735e1783063b7

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 35m 38s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 21m 25s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 29m 07s
adoption-standalone-to-crc-ceph-provider POST_FAILURE in 3h 03m 44s
✔️ openstack-operator-tempest-multinode SUCCESS in 1h 33m 41s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 1h 58m 54s

@stuggi

stuggi commented Sep 17, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

@abays abays left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Sep 18, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: abays, github-actions[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@abays

abays commented Sep 18, 2026

Copy link
Copy Markdown
Contributor

@github-actions: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:
Test name Commit Details Required Rerun command
ci/prow/openstack-operator-build-deploy-kuttl-4-20 3ca7cd1 link unknown /test openstack-operator-build-deploy-kuttl-4-20

Full PR test history. Your PR dashboard.
Details

{  failed to wait for the created cluster claim to become ready: timed out waiting for the condition}

/test openstack-operator-build-deploy-kuttl-4-20

@stuggi

stuggi commented Sep 18, 2026

Copy link
Copy Markdown
Contributor

/retest

@github-actions
github-actions Bot force-pushed the openstack-dependency-bump/main branch from 3ca7cd1 to ca26eb0 Compare September 19, 2026 10:07
@openshift-ci openshift-ci Bot removed the lgtm label Sep 19, 2026
@openshift-ci

openshift-ci Bot commented Sep 19, 2026

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants