Skip to content

openstack-k8s-operators dependency bump branch: 18.0-fr6 - #2049

Open
github-actions[bot] wants to merge 1 commit into
18.0-fr6from
openstack-dependency-bump/18.0-fr6
Open

github-actions[bot] wants to merge 1 commit into
18.0-fr6from
openstack-dependency-bump/18.0-fr6

Conversation

@github-actions

Copy link
Copy Markdown

Automated changes by create-pull-request GitHub action

@openshift-ci
openshift-ci Bot requested review from rabi and rebtoor August 22, 2026 10:06
@openshift-ci

openshift-ci Bot commented Aug 22, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: github-actions[bot]
Once this PR has been reviewed and has the lgtm label, please assign rabi for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Aug 22, 2026

Copy link
Copy Markdown
Contributor

Hi @github-actions[bot]. Thanks for your PR.

I'm waiting for a openstack-k8s-operators member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@github-actions
github-actions Bot force-pushed the openstack-dependency-bump/18.0-fr6 branch from 2ec7e4f to fc41b2b Compare August 29, 2026 10:06
@github-actions
github-actions Bot force-pushed the openstack-dependency-bump/18.0-fr6 branch from fc41b2b to 5103a5c Compare September 5, 2026 10:06
@centosinfra-prod-github-app

Copy link
Copy Markdown

Merge Failed.

This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.

@stuggi

stuggi commented Sep 12, 2026

Copy link
Copy Markdown
Contributor

/retest

@stuggi

stuggi commented Sep 12, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6fbf65d3e4134ae985f54393ec665b7f

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 12m 39s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 27m 51s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 29m 02s
adoption-standalone-to-crc-ceph-provider POST_FAILURE in 2h 59m 32s
✔️ openstack-operator-tempest-multinode SUCCESS in 1h 32m 18s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 1h 55m 03s

@stuggi

stuggi commented Sep 13, 2026

Copy link
Copy Markdown
Contributor

/retest

@stuggi

stuggi commented Sep 13, 2026

Copy link
Copy Markdown
Contributor

Cloning into 'install_yamls'...
fatal: unable to access 'https://github.com/openstack-k8s-operators/install_yamls.git/': Could not resolve host: github.com

@stuggi

stuggi commented Sep 13, 2026

Copy link
Copy Markdown
Contributor

/retest

@stuggi

stuggi commented Sep 13, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/d0c071f3b5414633a05918bfd7cac58a

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 15m 04s
podified-multinode-edpm-deployment-crc RETRY_LIMIT in 4m 23s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 38m 45s
adoption-standalone-to-crc-ceph-provider RETRY_LIMIT in 4m 21s
openstack-operator-tempest-multinode RETRY_LIMIT in 4m 32s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 2h 02m 40s

@stuggi

stuggi commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

recheck

@stuggi

stuggi commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

/test openstack-operator-build-deploy-kuttl-4-20

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c4fc0686b4cb4991a42155ab9e8f98e4

✔️ openstack-k8s-operators-content-provider SUCCESS in 2h 07m 11s
podified-multinode-edpm-deployment-crc RETRY_LIMIT in 4m 25s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 29m 21s
adoption-standalone-to-crc-ceph-provider RETRY_LIMIT in 4m 25s
✔️ openstack-operator-tempest-multinode SUCCESS in 1h 38m 27s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 1h 54m 17s

@stuggi

stuggi commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/5de665866f434bc4bac4d1bdc6d7541a

✔️ openstack-k8s-operators-content-provider SUCCESS in 3h 39m 59s
✔️ podified-multinode-edpm-deployment-crc SUCCESS in 1h 20m 58s
✔️ cifmw-crc-podified-edpm-baremetal SUCCESS in 1h 28m 51s
adoption-standalone-to-crc-ceph-provider POST_FAILURE in 3h 03m 05s
openstack-operator-tempest-multinode NODE_FAILURE Node(set) request 099-0000204616 failed in 0s
✔️ openstack-operator-edpm-baremetal-minor-update SUCCESS in 1h 57m 00s

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant