Skip to content

test/system: Silence SC2154, remove unused variable, etc. - #1840

Merged
debarshiray merged 3 commits into
containers:mainfrom
debarshiray:wip/rishi/test-system-dbus-arch-line-wrap
Sep 8, 2026
Merged

test/system: Silence SC2154, remove unused variable, etc.#1840
debarshiray merged 3 commits into
containers:mainfrom
debarshiray:wip/rishi/test-system-dbus-arch-line-wrap

Conversation

@debarshiray

@debarshiray debarshiray commented Sep 5, 2026

Copy link
Copy Markdown
Member

No description provided.

debarshiray added a commit to debarshiray/toolbox that referenced this pull request Sep 5, 2026
@debarshiray
debarshiray force-pushed the wip/rishi/test-system-dbus-arch-line-wrap branch from 7fe8e5a to 7429c92 Compare September 5, 2026 12:11
@centosinfra-prod-github-app

Copy link
Copy Markdown

Merge Failed.

This change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.

@debarshiray

Copy link
Copy Markdown
Member Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/5fbe92474ffe4ad2adfd9ccc9fcf0def

unit-test NODE_FAILURE Node(set) request 200-0000193306 failed in 0s
unit-test-migration-path-for-coreos-toolbox NODE_FAILURE Node(set) request 200-0000193307 failed in 0s
unit-test-restricted NODE_FAILURE Node(set) request 200-0000193308 failed in 0s
system-test-fedora-rawhide-commands-options NODE_FAILURE Node(set) request 200-0000193309 failed in 0s
system-test-fedora-rawhide-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193310 failed in 0s
system-test-fedora-rawhide-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193311 failed in 0s
system-test-fedora-44-commands-options NODE_FAILURE Node(set) request 200-0000193312 failed in 0s
system-test-fedora-44-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193313 failed in 0s
system-test-fedora-44-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193314 failed in 0s
system-test-fedora-43-commands-options NODE_FAILURE Node(set) request 200-0000193315 failed in 0s
system-test-fedora-43-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193316 failed in 0s
system-test-fedora-43-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193317 failed in 0s
system-test-fedora-42-commands-options NODE_FAILURE Node(set) request 200-0000193318 failed in 0s
system-test-fedora-42-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193319 failed in 0s
system-test-fedora-42-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193320 failed in 0s
system-test-fedora-coreos-next-commands-options NODE_FAILURE Node(set) request 200-0000193321 failed in 0s
system-test-fedora-coreos-next-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193322 failed in 0s
system-test-fedora-coreos-next-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193323 failed in 0s
system-test-fedora-coreos-stable-commands-options NODE_FAILURE Node(set) request 200-0000193324 failed in 0s
system-test-fedora-coreos-stable-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193325 failed in 0s
system-test-fedora-coreos-stable-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193326 failed in 0s
system-test-fedora-coreos-testing-commands-options NODE_FAILURE Node(set) request 200-0000193327 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-arch-fedora NODE_FAILURE Node(set) request 200-0000193328 failed in 0s
system-test-fedora-coreos-testing-runtime-environment-ubuntu NODE_FAILURE Node(set) request 200-0000193329 failed in 0s

@debarshiray

Copy link
Copy Markdown
Member Author

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/5a1a4540988749aea8f90ef8dcb24c39

✔️ unit-test SUCCESS in 2m 19s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 48s
✔️ unit-test-restricted SUCCESS in 2m 30s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 1h 02m 31s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 46m 02s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 9m 58s
✔️ system-test-fedora-44-commands-options SUCCESS in 43m 23s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 34m 05s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 10m 44s
✔️ system-test-fedora-43-commands-options SUCCESS in 44m 35s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 35m 30s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 10m 36s
✔️ system-test-fedora-42-commands-options SUCCESS in 38m 53s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 30m 56s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 9m 45s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 1h 02m 10s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 52m 54s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 15m 40s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 1h 01m 24s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 50m 56s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 12m 07s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 46m 10s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 36m 38s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 12m 05s

@debarshiray
debarshiray force-pushed the wip/rishi/test-system-dbus-arch-line-wrap branch from 7429c92 to 07906dd Compare September 8, 2026 10:22
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/f18a355d6dc54d9c99e507258941b4f7

✔️ unit-test SUCCESS in 2m 16s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 45s
✔️ unit-test-restricted SUCCESS in 2m 08s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 49m 43s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 29m 46s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 7m 23s
✔️ system-test-fedora-44-commands-options SUCCESS in 47m 57s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 37m 25s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 9m 40s
✔️ system-test-fedora-43-commands-options SUCCESS in 37m 41s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 37m 35s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 9m 25s
✔️ system-test-fedora-42-commands-options SUCCESS in 32m 59s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 23m 56s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 8m 28s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 50m 10s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 43m 20s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 12m 43s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 50m 26s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 42m 49s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 12m 38s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 49m 40s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 42m 27s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 12m 24s

Otherwise https://www.shellcheck.net/ would complain:
  Line 287:
  run echo "$name"
            ^---^ SC2154 (warning): name is referenced but not assigned.

See: https://www.shellcheck.net/wiki/SC2154

Note that there's no need to use Bats' 'run' helper to merely check if
the command succeeded or not, because 'set -e' is set for all tests [1].

[1] https://bats-core.readthedocs.io/en/stable/writing-tests.html

containers#1840
@debarshiray debarshiray changed the title test/system: Style fix test/system: Silence SC2154, remove unused variable, etc. Sep 8, 2026
@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/e3aad4168eaf450088be8281635fb21a

✔️ unit-test SUCCESS in 2m 15s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 10s
✔️ unit-test-restricted SUCCESS in 2m 10s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 39m 39s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 28m 53s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 7m 47s
✔️ system-test-fedora-44-commands-options SUCCESS in 38m 37s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 29m 33s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 7m 00s
✔️ system-test-fedora-43-commands-options SUCCESS in 38m 45s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 35m 38s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 9m 18s
✔️ system-test-fedora-42-commands-options SUCCESS in 39m 38s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 31m 18s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 6m 27s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 47m 50s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 39m 31s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 11m 42s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 47m 48s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 39m 40s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 11m 59s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 47m 27s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 39m 17s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 11m 40s

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build succeeded.
https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/local/buildset/4b40d40abf3049a5afab18df3c5fbcf8

✔️ unit-test SUCCESS in 2m 15s
✔️ unit-test-migration-path-for-coreos-toolbox SUCCESS in 3m 21s
✔️ unit-test-restricted SUCCESS in 2m 12s
✔️ system-test-fedora-rawhide-commands-options SUCCESS in 54m 18s
✔️ system-test-fedora-rawhide-runtime-environment-arch-fedora SUCCESS in 43m 06s
✔️ system-test-fedora-rawhide-runtime-environment-ubuntu SUCCESS in 9m 44s
✔️ system-test-fedora-44-commands-options SUCCESS in 52m 24s
✔️ system-test-fedora-44-runtime-environment-arch-fedora SUCCESS in 28m 12s
✔️ system-test-fedora-44-runtime-environment-ubuntu SUCCESS in 11m 04s
✔️ system-test-fedora-43-commands-options SUCCESS in 55m 04s
✔️ system-test-fedora-43-runtime-environment-arch-fedora SUCCESS in 43m 09s
✔️ system-test-fedora-43-runtime-environment-ubuntu SUCCESS in 7m 50s
✔️ system-test-fedora-42-commands-options SUCCESS in 32m 34s
✔️ system-test-fedora-42-runtime-environment-arch-fedora SUCCESS in 38m 48s
✔️ system-test-fedora-42-runtime-environment-ubuntu SUCCESS in 9m 16s
✔️ system-test-fedora-coreos-next-commands-options SUCCESS in 56m 19s
✔️ system-test-fedora-coreos-next-runtime-environment-arch-fedora SUCCESS in 47m 40s
✔️ system-test-fedora-coreos-next-runtime-environment-ubuntu SUCCESS in 13m 51s
✔️ system-test-fedora-coreos-stable-commands-options SUCCESS in 55m 49s
✔️ system-test-fedora-coreos-stable-runtime-environment-arch-fedora SUCCESS in 48m 44s
✔️ system-test-fedora-coreos-stable-runtime-environment-ubuntu SUCCESS in 14m 14s
✔️ system-test-fedora-coreos-testing-commands-options SUCCESS in 56m 19s
✔️ system-test-fedora-coreos-testing-runtime-environment-arch-fedora SUCCESS in 46m 30s
✔️ system-test-fedora-coreos-testing-runtime-environment-ubuntu SUCCESS in 13m 44s

@debarshiray
debarshiray merged commit 586b88f into containers:main Sep 8, 2026
3 checks passed
@debarshiray
debarshiray deleted the wip/rishi/test-system-dbus-arch-line-wrap branch September 8, 2026 14:10
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant