Update ATTRIBUTION.txt file - #10943
eks-distro-pr-bot wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @eks-distro-pr-bot. Thanks for your PR. I'm waiting for a aws member to verify that this patch is reasonable to test. If it is, they should reply with Tip We noticed you've done this a few times! Consider joining the org to skip this step and gain Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
3a8c533 to
f2a8819
Compare
ee50ace to
f7b60b9
Compare
4e0326d to
af54ef3
Compare
493981d to
7e851ba
Compare
34dbe33 to
bfb165e
Compare
bfb165e to
535a773
Compare
This PR updates the ATTRIBUTION.txt file if there have been changes.
By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.